Skip to content

Update CODEOWNERS to change team assignments - #337

Merged
korniltsev-grafanista merged 1 commit into
mainfrom
korniltsev-grafanista-patch-2
Jul 17, 2026
Merged

korniltsev-grafanista merged 1 commit into
mainfrom
korniltsev-grafanista-patch-2

Conversation

@korniltsev-grafanista

Copy link
Copy Markdown
Contributor

No description provided.

@korniltsev-grafanista
korniltsev-grafanista requested a review from a team as a code owner July 14, 2026 08:11
@korniltsev-grafanista
korniltsev-grafanista merged commit 9515fce into main Jul 17, 2026
73 checks passed
@korniltsev-grafanista
korniltsev-grafanista deleted the korniltsev-grafanista-patch-2 branch July 17, 2026 03:01
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants