Skip to content

feat!: added kgrep/kgrepn - #109

Open
eshepelyuk wants to merge 1 commit into
kubermatic:mainfrom
eshepelyuk:feat/kgrep
Open

feat!: added kgrep/kgrepn#109
eshepelyuk wants to merge 1 commit into
kubermatic:mainfrom
eshepelyuk:feat/kgrep

Conversation

@eshepelyuk

Copy link
Copy Markdown
Contributor
  • Search for a string in resources all namespaces / current namespace
  • Removed broken ksearch

Resolves #91
Resolves #95

@kubermatic-bot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign scheeles for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubermatic-bot kubermatic-bot added dco-signoff: no Denotes that at least one commit in the pull request doesn't have a valid DCO signoff message. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. labels Aug 13, 2026
* Search for a string in resources all namespaces / current namespace
* Removed broken ksearch

Resolves kubermatic#91
Resolves kubermatic#95

Signed-off-by: Ievgenii Shepeliuk <eshepelyuk@gmail.com>
@kubermatic-bot kubermatic-bot added dco-signoff: yes Denotes that all commits in the pull request have the valid DCO signoff message. and removed dco-signoff: no Denotes that at least one commit in the pull request doesn't have a valid DCO signoff message. labels Aug 13, 2026
@eshepelyuk

Copy link
Copy Markdown
Contributor Author

/assign @dharapvj
/assign @mfahlandt
/assign @ronissac88
/assign @scheeles
/assign @toschneck
/assign @xrstf

@eshepelyuk

Copy link
Copy Markdown
Contributor Author

@xrstf could you plz take a look ?

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dco-signoff: yes Denotes that all commits in the pull request have the valid DCO signoff message. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

ksearch error: unknown flag: --export

8 participants