Conversation
This test repeatedly fails on go 1.26. The feature was tech preview and deprecated ref: https://redhat.atlassian.net/browse/ACM-35531 Signed-off-by: Janelle Law <jalaw@redhat.com>
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository: stolostron/coderabbit/.coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: jan-law The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/retest Pasting latest failure (flake?) for reference |
|
/retest |
|
/retest |
|
/retest |
|
@jan-law: The following test failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
I'm not convinced the compliance history test contributed to the failures in #1092 The latest retest failed with a different error |
Unblocks #1092
See discussion starting at #1092 (comment)