Skip to content

https://redhat.atlassian.net/browse/ACM-34545 JTBD 1st attempt gov - #9097

Merged
dockerymick merged 13 commits into
2.17_stagefrom
mj-jtb-gov
Sep 14, 2026
Merged

dockerymick merged 13 commits into
2.17_stagefrom
mj-jtb-gov

Conversation

@dockerymick

Copy link
Copy Markdown
Contributor

Started with Jobs 1-7

Comment thread governance/config_policy_ctrl.adoc Outdated
* link:https://kubernetes.io/docs/reference/using-api/api-concepts/#dry-run[Dry-run]
* xref:../governance/supported_policies.adoc#supported-policies[Supported policies]
* xref:../governance/policy_controllers.adoc#policy-controllers[Policy controllers]
* xref:../governance/policy_controllers.adoc#policy-controllers[Understand and configure policy controllers]

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Discussing this, most likely will change


[#generate-policy-comp-operator]
= Generating a policy that installs the Compliance Operator
= Compliance Operator installation example

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Need to update this (Procedure)

Comment thread governance/grc_intro.adoc Outdated

[#governance]
= Governance
= Understand the governance framework

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Most likely change the title

Comment thread governance/grc_intro.adoc Outdated
* xref:../governance/policy_controllers_intro.adoc#policy-controllers[Policy controllers]
* xref:../governance/template_support_intro.adoc#template-processing[Template processing]
* xref:../governance/policy_deployment.adoc#policy-deployment[Choose a policy deployment strategy]
* xref:../governance/policy_controllers_intro.adoc#policy-controllers[Understand and configure policy controllers]

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Will update based on feedback

Comment thread governance/hub_policy_framework.adoc Outdated

[#hub-policy-framework]
= Hub cluster policy framework
= Deploy policies using the hub framework

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Starting with verb but this is a concept. Need to improve

Comment thread governance/policy_generator.adoc Outdated

[#policy-generator]
= Policy Generator
= Policy Generator configuration and reference

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Remove the (..and reference)

Comment thread governance/policy_generator.adoc Outdated
* link:../applications/subscription_sample.adoc#applying-kustomize[Applying Kustomize]
* link:../gitops/gitops_overview.adoc#gitops-overview[GitOps overview]
* xref:../governance/grc_intro.adoc#governance[Governance]
* xref:../governance/grc_intro.adoc#governance[Understand the governance framework]

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Update title. We do not want to use "Understand" for a concept

Comment thread governance/template_functions.adoc Outdated

[#template-functions]
= Template functions
= Template functions reference

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Remove "Reference"

Comment thread governance/template_functions.adoc Outdated
* xref:../governance/template_function_sprig.adoc#sprig-open-source[Sprig open source]
* xref:../governance/template_support_intro.adoc#template-processing[Template processing]
* xref:../governance/adv_template_process.adoc#adv-template-processing[Advanced template processing in policies]
* xref:../governance/template_support_intro.adoc#template-processing[Use template processing in policies]

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Need to update title

Comment thread governance/template_support_intro.adoc Outdated

[#template-processing]
= Template processing
= Use template processing in policies

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Starting with a verb, but this is a concept. Update

@JustinKuli JustinKuli left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM

@openshift-ci

openshift-ci Bot commented Aug 25, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: dockerymick, JustinKuli

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Sep 4, 2026

Copy link
Copy Markdown

New changes are detected. LGTM label has been removed.

@ch-stark

Copy link
Copy Markdown

lgtm

@dockerymick
dockerymick merged commit 33df407 into 2.17_stage Sep 14, 2026
3 of 4 checks passed
@dockerymick
dockerymick deleted the mj-jtb-gov branch September 14, 2026 14:11
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants